chore: update Quarkus platform version to 3.39.5 - #4062
Conversation
|
Hi @github-actions[bot]. Thanks for your PR. I'm waiting for a knative member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: github-actions[bot], lkingland The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #4062 +/- ##
==========================================
+ Coverage 56.73% 56.74% +0.01%
==========================================
Files 218 218
Lines 25994 25994
==========================================
+ Hits 14748 14751 +3
+ Misses 9902 9900 -2
+ Partials 1344 1343 -1
Flags with carried forward coverage won't be shown. Click here to find out more. ☔ View full report in Codecov by Harness. 🚀 New features to boost your workflow:
|
chore: update Quarkus platform version to 3.39.5